Skip to content

refactor, add xtables.ExecWrapper - #18

Merged
danwinship merged 5 commits into
kubernetes-sigs:masterfrom
danwinship:refactor
Sep 19, 2026
Merged

danwinship merged 5 commits into
kubernetes-sigs:masterfrom
danwinship:refactor

Conversation

@danwinship

Copy link
Copy Markdown
Contributor

This cleans up the DetectMode code and adds the new xtables.ExecWrapper, to (eventually) allow wrapping utiliptables with automatic legacy/nft detection

/hold
want to make sure CI runs as expected

@kubernetes-prow kubernetes-prow Bot added do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Sep 18, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from aojea and thockin September 18, 2026 19:50
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Sep 18, 2026
@danwinship

Copy link
Copy Markdown
Contributor Author

/hold cancel

@kubernetes-prow kubernetes-prow Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 18, 2026
@danwinship

Copy link
Copy Markdown
Contributor Author

(running apidiff for informational purposes but it's intentionally non-required)

Comment thread pkg/xtables/detect.go
Comment on lines -38 to -39
xtablesLegacyMultiBinaryName = "xtables-legacy-multi"
xtablesNFTMultiBinaryName = "xtables-nft-multi"

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

yeah, this was always surprising my why we depend on those

@aojea

aojea commented Sep 19, 2026

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 19, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: aojea, danwinship

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@danwinship
danwinship merged commit c80f8ab into kubernetes-sigs:master Sep 19, 2026
3 of 5 checks passed
@danwinship
danwinship deleted the refactor branch September 19, 2026 13:18
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants