Skip to content

Fix make lint install path - #123

Open
stmcginnis wants to merge 1 commit into
kubernetes-sigs:mainfrom
stmcginnis:fix-make-lint
Open

stmcginnis wants to merge 1 commit into
kubernetes-sigs:mainfrom
stmcginnis:fix-make-lint

Conversation

@stmcginnis

Copy link
Copy Markdown
Contributor

When we moved to golangci-lint v2, the go install path changed to point to the v2 module path. This worked fine in CI since the golangci-lint action automatically takes care of it based on the configured version. But attempting to run make install on a machine that doesn't already have the binary results in a failure.

This updates the defaulted path for the linter to be the correct current v2 path.

Also bumps up to the next point release while we're at it.

When we moved to golangci-lint v2, the `go install` path changed to
point to the v2 module path. This worked fine in CI since the
golangci-lint action automatically takes care of it based on the
configured version. But attempting to run `make install` on a machine
that doesn't already have the binary results in a failure.

This updates the defaulted path for the linter to be the correct current
v2 path.

Also bumps up to the next point release while we're at it.

Signed-off-by: Sean McGinnis <sean.mcginnis@gmail.com>
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: stmcginnis

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the sig/release Categorizes an issue or PR as relevant to SIG Release. label Aug 10, 2026
@kubernetes-prow kubernetes-prow Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Aug 10, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. sig/release Categorizes an issue or PR as relevant to SIG Release. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant