Conversation
|
The latest updates on your projects. Learn more about Vercel for GitHub. 1 Skipped Deployment
|
hannojg
marked this pull request as draft
September 26, 2026 19:58
hannojg
force-pushed
the
hannojg/fix-nitro-view-dynamic-props
branch
from
September 26, 2026 22:15
cd8c32b to
bbc289c
Compare
hannojg
changed the base branch from
main
to
hannojg/nitro-android-props-build-flag
September 26, 2026 22:16
hannojg
force-pushed
the
hannojg/fix-nitro-view-dynamic-props
branch
from
September 26, 2026 22:24
bbc289c to
4101400
Compare
hannojg
force-pushed
the
hannojg/fix-nitro-view-dynamic-props
branch
from
September 26, 2026 22:29
4101400 to
05eb5cd
Compare
hannojg
force-pushed
the
hannojg/fix-nitro-view-dynamic-props
branch
from
September 26, 2026 22:35
05eb5cd to
5e92d2e
Compare
hannojg
added this pull request to stack #1664
September 26, 2026 22:40
This branch was previously deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
In:
enableCppPropsIteratorSetterthrough a copy ctor +RawProps::forEachItem(#57328) react/react-native#57328meta removed calling
initializeDynamicProps(..)in the Props constructor. For their purposes this was fine, however, we rely on this being called to initialize thefolly::dynamic rawPropsinsideProps.h.This PR:
Fixes:
For making sure AI doesn't get prop handling wrong in the future i added this test, which should be merged as well: