feat(eth-lc): electra fork - #3878
Closed
hussein-aitlahcen wants to merge 3 commits into
Closed
Conversation
|
The latest updates on your projects. Learn more about Vercel for Git ↗︎ 1 Skipped Deployment
|
hussein-aitlahcen
force-pushed
the
ethereum-electra
branch
3 times, most recently
from
February 26, 2025 17:13
41de505 to
8be7bf4
Compare
hussein-aitlahcen
force-pushed
the
ethereum-electra
branch
from
February 26, 2025 17:15
8be7bf4 to
249756f
Compare
benluelo
reviewed
Feb 26, 2025
benluelo
added a commit
that referenced
this pull request
Mar 10, 2025
massive refactor to our beacon chain types. we are now aware of all forks and forced to handle them. also versioned the ethereum client state, potentially will do the same to the consensus state? includes the changes from #3878 as well. TODO: - [ ] docs on new libraries - [ ] json spec tests for the beacon types (not a hard requirement for merging this pr)
Contributor
|
superceded by #3995 |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR introduces the new constants and refactor the existing LC to handle electra fork.
See https://github.com/ethereum/consensus-specs/blob/dev/specs/electra/light-client/sync-protocol.md